Repository navigation
fix: Rename inner module from argoproj-labs to redhat-developer - #1334
olivergondza wants to merge 2 commits into
Conversation
This was named incorrectly during the merge. ```diff - "github.com/argoproj-labs/gitops-operator/argocd-operator/XXX" + "github.com/redhat-developer/gitops-operator/argocd-operator/XXX" ``` Signed-off-by: Oliver Gondža <ogondza@gmail.com>
|
Important Review skippedToo many files! This PR contains 418 files, which is 118 over the limit of 300. To get a review, reduce the PR to 300 files or fewer by splitting it into smaller PRs or changing its base branch. ⚙️ Run configuration
📒 Files selected for processing (418)
You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
/test v4.19-e2e |
|
/test v4.14-kuttl-parallel |
|
|
||
| // Internalized from a separate repo | ||
| require github.com/argoproj-labs/gitops-operator/argocd-operator v0.20.0-rc1 | ||
| require github.com/redhat-developer/gitops-operator/argocd-operator v0.20.0-rc1 |
There was a problem hiding this comment.
the above tag does not exist yet. Now that 1.22.0 is out, I think we should cut the release in the project to make v1.22.0 available. It is not problem since this dep is replaced with the local copy below, but it's odd to point to a non-exist tag.
There was a problem hiding this comment.
This is the tag the argocd-operator was forked at. It kind of makes sense to me to keep it to indicate when the codebase has split.
You are correct it does not matter what release version we specify when it is replaced.
I am not sure if this makes sense changing to any tag from gitops-operator repo. Unless we commit to keep on updating it to latest gitops-operator tag, it will get stale and point to some random historical tag causing confusion, IMO.
|
you may want to double check some dot file and dot dirs for any occurrences of argoproj-labs. Files like .coderabbit.yaml, .golangci-lint.yml, .github/** |
Yeah, only the |
|
Can we not add a |
Sorry, I am not sure I am following. What would be the purpose? |
|
/lgtm |
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: anandf The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
/test unit |
…into fix-module-name Signed-off-by: Oliver Gondža <ogondza@gmail.com>
|
New changes are detected. LGTM label has been removed. |
|
@olivergondza: The following tests failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
PR needs rebase. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
What type of PR is this?
What does this PR do / why we need it:
Use the module name corresponding with the org/repository name.
This was named incorrectly during the merge.
Have you updated the necessary documentation?
Which issue(s) this PR fixes:
Fixes #?
Test acceptance criteria:
How to test changes / Special notes to the reviewer: